Conversation
pre-commit.check-secrets: ENABLED
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: agullon The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test e2e-aws-tests-bootc-release |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
/retest |
|
/test e2e-aws-tests-release |
1 similar comment
|
/test e2e-aws-tests-release |
|
/retest |
|
@agullon: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
Please, do not restest the CI test jobs again. Some scenarios are flaky and failed, but they passed after a retest. Unfortunatelly there's always at least 1 failure scenario on every CI test job. I'll work on adress this later. |
|
Automated release testing for 4.21.31 completed. Artifacts uploaded to S3. |
Release testing PR for MicroShift 4.21.31. Do not merge.
Created by /microshift-release:automated-testing