NO-ISSUE: rebase-release-5.1-5.1.0-0.nightly-2026-08-27-012048_amd64-2026-08-27_arm64-2026-08-25 - #7273
Conversation
|
@microshift-rebase-script[bot]: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@microshift-rebase-script[bot]: This pull request explicitly references no jira issue. Retaining the jira/valid-bug label as it was manually added. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: microshift-rebase-script[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Extra messages:
|
|
/test e2e-aws-ai-model-serving Added by prow-jobs-for-pull-requests.sh 🤖 |
bca006e to
d51d83b
Compare
|
Rebase job updated the branch amd64: 5.1.0-0.nightly-2026-08-27-012048
The change list was truncated. See scripts/auto-rebase/changelog.txt in the PR for the full details. /label tide/merge-method-squash Extra messages:
|
|
@microshift-rebase-script[bot]: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/close |
|
@pacevedom: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
amd64: 5.1.0-0.nightly-2026-08-27-012048
arm64: 5.1.0-0.nightly-arm64-2026-08-25-194037
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-5.1-periodics-rebase-on-nightlies/2094274194471653376
The change list was truncated. See scripts/auto-rebase/changelog.txt in the PR for the full details.
/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug