NO-ISSUE: rebase-release-5.0-5.0.0-0.nightly-2026-08-26-175402_amd64-2026-08-26_arm64-2026-08-26 - #7255
Conversation
|
@microshift-rebase-script[bot]: This pull request explicitly references no jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@microshift-rebase-script[bot]: This pull request explicitly references no jira issue. Retaining the jira/valid-bug label as it was manually added. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: microshift-rebase-script[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test e2e-aws-ai-model-serving Added by prow-jobs-for-pull-requests.sh 🤖 |
|
@microshift-rebase-script[bot]: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/hold Revision 1e40e51 was retested 3 times: holding |
|
Closing as duplicate: superseded by #7272. |
|
@microshift-rebase-script[bot]: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
amd64: 5.0.0-0.nightly-2026-08-26-175402
arm64: 5.0.0-0.nightly-arm64-2026-08-26-175351
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-5.0-periodics-rebase-on-nightlies/2092824864057462784
api embedded-component 7e228f5314e1b6db6ff2370bdaac4583ebc306b2 to 98e1b9eebf85902920e836ef1683b63d28af28c0
cluster-ingress-operator embedded-component acca9642a81370bc1a587155f7e1e7998b7c5489 to 7e78dfcc48612331125e07cb886d64c3296ee192
cluster-kube-apiserver-operator embedded-component 317cbc15caca68381bb89f9c82e279750b443c58 to 440392249bc58324156ba813b797728e9a81fa3c
cluster-kube-controller-manager-operator embedded-component 8db74e3fe8793043d942ef1f59cce3b0a0bcc548 to 7be80f36a759d713170588c44fcd5b45970382d5
cluster-network-operator embedded-component 4d1d5c53bdce02b4b95c6b03b8c08f2b60280a1f to 0813f5cf408bcd19d4123cb7a9531772a7957b0c
cluster-openshift-controller-manager-operator embedded-component 34f95b07f4afbc47558e54e4fa2710fd692e615e to 8da2f1fcb1e76e8b1b97b16ca7bbfa7116287eb8
machine-config-operator embedded-component 0df05f6ea615ec212f68633f92d01f6e248d2a34 to 27ba14fb2a77d6ee08d4ec0f450bc24f69b1e5cf
openshift-controller-manager embedded-component 5631cf493b006cbc72a8600a7435813272d71940 to 726a0562818b68faccfa130ae7ea17ff42915418
operator-framework-olm embedded-component c64b9ca2026d13e8907d547b1cbe5226b0a25219 to c3d56d2021bbb8d57bc359fabded3b35b253f55e
/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug