Update registry.access.redhat.com/ubi9/openjdk-21 Docker tag to v1.24-3.1789373067 - #669
Conversation
|
Hi @red-hat-konflux[bot]. Thanks for your PR. I'm waiting for a devfile member to verify that this patch is reasonable to test. If it is, they should reply with Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/ok-to-test |
17a05c8 to
c831f8a
Compare
c831f8a to
d27a540
Compare
d27a540 to
e248b7a
Compare
e248b7a to
0028968
Compare
0028968 to
7494000
Compare
7494000 to
a6e0c9f
Compare
a6e0c9f to
5ad9e3a
Compare
5ad9e3a to
bf317e0
Compare
bf317e0 to
8ee3556
Compare
8ee3556 to
632afc6
Compare
632afc6 to
cab82d5
Compare
cab82d5 to
b58c0ab
Compare
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
…-3.1789373067 Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
b58c0ab to
e022d1b
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux[bot], svor The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
This PR contains the following updates:
1.24-2→1.24-3.1789373067Warning
Some dependencies could not be looked up. Check the warning logs for more information.
Configuration
📅 Schedule: (in timezone America/Toronto)
* 19-23,0-2 */2,4 * 4)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.